Skip to content

Add pr-approve script - #821

Open
mbaldessari wants to merge 1 commit into
validatedpatterns:mainfrom
mbaldessari:pr-approve-script
Open

Add pr-approve script#821
mbaldessari wants to merge 1 commit into
validatedpatterns:mainfrom
mbaldessari:pr-approve-script

Conversation

@mbaldessari

Copy link
Copy Markdown
Contributor

No description provided.

@openshift-ci

openshift-ci Bot commented Sep 11, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: mbaldessari

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant