Skip to content

Update CODEOWNERS file for team review - #801

Merged
alexgromero merged 2 commits into
smithy-lang:developfrom
alexgromero:update-codeowners
Sep 17, 2026
Merged

alexgromero merged 2 commits into
smithy-lang:developfrom
alexgromero:update-codeowners

Conversation

@alexgromero

Copy link
Copy Markdown
Contributor

Issue #, if available:

Description of changes:
Update .github/CODEOWNERS to require team review from core SDK team on all PRs.

By submitting this pull request, I confirm that you can use, modify, copy, and redistribute this contribution, under the terms of your choice.

@alexgromero
alexgromero requested a review from a team as a code owner September 17, 2026 18:42
arandito
arandito previously approved these changes Sep 17, 2026
@alexgromero
alexgromero requested review from a team and arandito and removed request for a team September 17, 2026 19:59
@alexgromero
alexgromero merged commit c5b1e37 into smithy-lang:develop Sep 17, 2026
8 checks passed
@alexgromero
alexgromero deleted the update-codeowners branch September 17, 2026 20:33
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants