🌱 bump go.opentelemetry.io/otel/exporters/otlp/otlptrace/otlptracehttp from 1.44.0 to 1.45.0 - #2933
Conversation
✅ Deploy Preview for olmv1 ready!
To edit notification comments on pull requests, go to your Netlify project configuration. |
|
Important Review skippedBot user detected. To trigger a single review, invoke the ⚙️ Run configurationConfiguration used: defaults Review profile: CHILL Plan: Advanced Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
Bumps [go.opentelemetry.io/otel/exporters/otlp/otlptrace/otlptracehttp](https://github.com/open-telemetry/opentelemetry-go) from 1.44.0 to 1.45.0. - [Release notes](https://github.com/open-telemetry/opentelemetry-go/releases) - [Changelog](https://github.com/open-telemetry/opentelemetry-go/blob/main/CHANGELOG.md) - [Commits](open-telemetry/opentelemetry-go@v1.44.0...v1.45.0) --- updated-dependencies: - dependency-name: go.opentelemetry.io/otel/exporters/otlp/otlptrace/otlptracehttp dependency-version: 1.45.0 dependency-type: indirect ... Signed-off-by: dependabot[bot] <support@github.com>
7507177 to
2028564
Compare
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: perdasilva The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
Bumps go.opentelemetry.io/otel/exporters/otlp/otlptrace/otlptracehttp from 1.44.0 to 1.45.0.
Release notes
Sourced from go.opentelemetry.io/otel/exporters/otlp/otlptrace/otlptracehttp's releases.
... (truncated)
Changelog
Sourced from go.opentelemetry.io/otel/exporters/otlp/otlptrace/otlptracehttp's changelog.
... (truncated)
Commits
93a693eRelease v1.45.0 (#8693)c65d435Merge commit from fork223f9fdsdk/metric: remove obsolete randomFloat64 TODO (#8685)06272bcfix(deps): update googleapis to 6ac0973 (#8694)a4f238fchore(deps): update github.com/charmbracelet/ultraviolet digest to 8b69304 (#...37140e7chore(deps): update codspeedhq/action action to v5.0.2 (#8690)cef0855chore(deps): update module github.com/lucasb-eyer/go-colorful to v1.4.1 (#8689)e814a72Merge commit from forkbfd8eb7chore(deps): update github.com/golangci/rowserrcheck digest to d2031e3 (#8687)48db2c6chore(deps): update github/codeql-action action to v4.37.5 (#8692)