Skip to content

Add CODEOWNERS file - #43

Open
derkat8992 wants to merge 1 commit into
mainfrom
add-codeowners
Open

derkat8992 wants to merge 1 commit into
mainfrom
add-codeowners

Conversation

@derkat8992

Copy link
Copy Markdown

Adds a CODEOWNERS file to define default code owners for all files in the repository.

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant