Skip to content

Replace TOC with SC in OWNERS file - #16699

Open
aliok wants to merge 1 commit into
mainfrom
aliok-patch-1
Open

aliok wants to merge 1 commit into
mainfrom
aliok-patch-1

Conversation

@aliok

@aliok aliok commented Oct 7, 2026

Copy link
Copy Markdown
Member

This pull request updates the OWNERS file to reflect a change in the approver group for the repository.

Ownership updates:

  • Replaces technical-oversight-committee with steering-committee in the list of approvers in the pkg/apis/OWNERS file.Fixes #

Proposed Changes

Release Note


@knative-prow
knative-prow Bot requested review from dprotaso and dsimansk October 7, 2026 08:50
@knative-prow

knative-prow Bot commented Oct 7, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@knative-prow knative-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Oct 7, 2026
@aliok

aliok commented Oct 7, 2026

Copy link
Copy Markdown
Member Author

cc @knative/serving-approvers @knative/steering-committee

@aliok aliok changed the title Update approvers in OWNERS file Replace TOC with SC in OWNERS file Oct 7, 2026
@codecov

codecov Bot commented Oct 7, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 80.37%. Comparing base (47ddb67) to head (ce4538f).

Additional details and impacted files
@@            Coverage Diff             @@
##             main   #16699      +/-   ##
==========================================
+ Coverage   80.32%   80.37%   +0.05%     
==========================================
  Files         217      217              
  Lines       13560    13560              
==========================================
+ Hits        10892    10899       +7     
+ Misses       2303     2299       -4     
+ Partials      365      362       -3     

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@dsimansk

dsimansk commented Oct 7, 2026

Copy link
Copy Markdown
Contributor

/approve
/lgtm

@knative-prow knative-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Oct 7, 2026
@knative-prow

knative-prow Bot commented Oct 7, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: aliok, dsimansk
Once this PR has been reviewed and has the lgtm label, please assign dprotaso for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@dsimansk

dsimansk commented Oct 7, 2026

Copy link
Copy Markdown
Contributor

/override "istio-latest-no-mesh-tls_serving_main"

@knative-prow

knative-prow Bot commented Oct 7, 2026

Copy link
Copy Markdown

@dsimansk: Overrode contexts on behalf of dsimansk: istio-latest-no-mesh-tls_serving_main

Details

In response to this:

/override "istio-latest-no-mesh-tls_serving_main"

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

lgtm Indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants